Skip to content

Bump postcss from 8.5.15 to 8.5.16 - #6

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/postcss-8.5.16
Closed

Bump postcss from 8.5.15 to 8.5.16#6
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/postcss-8.5.16

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jun 30, 2026

Copy link
Copy Markdown
Contributor

Bumps postcss from 8.5.15 to 8.5.16.

Release notes

Sourced from postcss's releases.

8.5.16

Changelog

Sourced from postcss's changelog.

8.5.16

Commits
Maintainer changes

This version was pushed to npm by GitHub Actions, a new releaser for postcss since your current version.


@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code labels Jun 30, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/postcss-8.5.16 branch from 24ee7a5 to 83b9b3f Compare July 9, 2026 12:26
@vercel

vercel Bot commented Jul 9, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
good-wallet-v2 Ignored Ignored Preview Jul 9, 2026 12:40pm

Request Review

Bumps [postcss](https://github.com/postcss/postcss) from 8.5.15 to 8.5.16.
- [Release notes](https://github.com/postcss/postcss/releases)
- [Changelog](https://github.com/postcss/postcss/blob/main/CHANGELOG.md)
- [Commits](postcss/postcss@8.5.15...8.5.16)

---
updated-dependencies:
- dependency-name: postcss
  dependency-version: 8.5.16
  dependency-type: direct:development
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/postcss-8.5.16 branch from 83b9b3f to f609026 Compare July 9, 2026 12:40
@dependabot @github

dependabot Bot commented on behalf of github Jul 14, 2026

Copy link
Copy Markdown
Contributor Author

Superseded by #21.

@dependabot dependabot Bot closed this Jul 14, 2026
@dependabot
dependabot Bot deleted the dependabot/npm_and_yarn/postcss-8.5.16 branch July 14, 2026 10:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants