Skip to content

Bump @toruslabs/customauth from 20.4.0 to 22.3.3 - #9

Closed
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/toruslabs/customauth-22.3.3
Closed

Bump @toruslabs/customauth from 20.4.0 to 22.3.3#9
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/toruslabs/customauth-22.3.3

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jun 30, 2026

Copy link
Copy Markdown
Contributor

Bumps @toruslabs/customauth from 20.4.0 to 22.3.3.

Release notes

Sourced from @​toruslabs/customauth's releases.

v22.3.3

What's Changed

Full Changelog: torusresearch/CustomAuth@v22.3.2...v22.3.3

v22.3.2

Changes

  • Add source parameter to CustomAuth (#471)

v22.3.1

What's Changed

Full Changelog: torusresearch/CustomAuth@v22.3.0...v22.3.1

v22.3.0

What's Changed

Full Changelog: torusresearch/CustomAuth@v22.2.0...v22.3.0

v22.2.0

What's Changed

Full Changelog: torusresearch/CustomAuth@v22.1.0...v22.2.0

v22.1.0

What's Changed

Full Changelog: torusresearch/CustomAuth@v22.0.0...v22.1.0

v22.0.0

What's Changed

... (truncated)

Commits
  • f6a261c Release 22.3.3
  • af1fc26 lock
  • b0144a6 Merge pull request #472 from torusresearch/feat/track-auth-flow
  • bfb9f26 feat: upgrade torus.js
  • 04a3adb chore: removed audit from CustomAuth
  • 63109a6 feat: save recordId in session manager for login redirect
  • d78983b chore: addressed cursor comments
  • 4f450cd feat: track 'oauthFailed', 'oauthInitiated' event for audit
  • 0383f31 Release 22.3.2
  • 091dfb7 Merge pull request #471 from torusresearch/chore/add-source
  • Additional commits viewable in compare view

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code labels Jun 30, 2026
@vercel

vercel Bot commented Jun 30, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated (UTC)
good-wallet-v2 Ignored Ignored Preview Jul 9, 2026 12:38pm

Request Review

@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/toruslabs/customauth-22.3.3 branch from 484960e to 640b60d Compare July 9, 2026 12:27
Bumps [@toruslabs/customauth](https://github.com/torusresearch/CustomAuth) from 20.4.0 to 22.3.3.
- [Release notes](https://github.com/torusresearch/CustomAuth/releases)
- [Commits](torusresearch/CustomAuth@v20.4.0...v22.3.3)

---
updated-dependencies:
- dependency-name: "@toruslabs/customauth"
  dependency-version: 22.3.3
  dependency-type: direct:production
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/toruslabs/customauth-22.3.3 branch from 640b60d to cacd4bb Compare July 9, 2026 12:38
@sunela-tr sunela-tr closed this Jul 14, 2026
@dependabot @github

dependabot Bot commented on behalf of github Jul 14, 2026

Copy link
Copy Markdown
Contributor Author

OK, I won't notify you again about this release, but will get in touch when a new version is available. If you'd rather skip all updates until the next major or minor version, let me know by commenting @dependabot ignore this major version or @dependabot ignore this minor version. You can also ignore all major, minor, or patch releases for a dependency by adding an ignore condition with the desired update_types to your config file.

If you change your mind, just re-open this PR and I'll resolve any conflicts on it.

@dependabot
dependabot Bot deleted the dependabot/npm_and_yarn/toruslabs/customauth-22.3.3 branch July 14, 2026 11:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant