build(deps): bump tags.cncf.io/container-device-interface from 1.1.0 to 1.1.1 - #2983
Merged
hami-robot[bot] merged 1 commit intoSep 9, 2026
Conversation
Bumps [tags.cncf.io/container-device-interface](https://github.com/cncf-tags/container-device-interface) from 1.1.0 to 1.1.1. - [Release notes](https://github.com/cncf-tags/container-device-interface/releases) - [Changelog](https://github.com/cncf-tags/container-device-interface/blob/main/RELEASE.md) - [Commits](cncf-tags/container-device-interface@v1.1.0...v1.1.1) --- updated-dependencies: - dependency-name: tags.cncf.io/container-device-interface dependency-version: 1.1.1 dependency-type: direct:production update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
dependabot
Bot
force-pushed
the
dependabot/go_modules/tags.cncf.io/container-device-interface-1.1.1
branch
from
September 9, 2026 02:20
db306ec to
8294c51
Compare
Contributor
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: dependabot[bot], Shouren The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
dependabot
Bot
deleted the
dependabot/go_modules/tags.cncf.io/container-device-interface-1.1.1
branch
September 9, 2026 03:28
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Bumps tags.cncf.io/container-device-interface from 1.1.0 to 1.1.1.
Release notes
Sourced from tags.cncf.io/container-device-interface's releases.
... (truncated)
Commits
cb98bcfMerge pull request #361 from elezar/bump-v1.1.13757e39Bump version to v1.1.173444d1Merge pull request #344 from thaJeztah/cdi_cleans65a7147pkg/cdi: ignore unrelated filesystem events73e8572pkg/cdi: handle fsnotify operations as a bitmask1a0198bpkg/cdi: watch for Create events on all platforms0656473pkg/cdi: simplify restoration of removed directory watches72f9410Merge pull request #358 from thaJeztah/cdi_cleans_step223b69d2Merge pull request #359 from thaJeztah/bump_fsnotifyd50fc58chore(deps): github.com/fsnotify/fsnotify v1.7.0