Skip to content

chore: remove public-api report from docs team's CODEOWNERS rule - #1012

Merged
barjin merged 1 commit into
masterfrom
chore/remove-api-report-from-codeowners
Aug 26, 2026
Merged

chore: remove public-api report from docs team's CODEOWNERS rule#1012
barjin merged 1 commit into
masterfrom
chore/remove-api-report-from-codeowners

Conversation

@barjin

@barjin barjin commented Aug 26, 2026

Copy link
Copy Markdown
Member

Related to #1008

@github-actions github-actions Bot added this to the 148th sprint - Tooling team milestone Aug 26, 2026
@github-actions github-actions Bot added the t-tooling Issues with this label are in the ownership of the tooling team. label Aug 26, 2026
@github-actions

Copy link
Copy Markdown
Contributor

⚠️ There are broken links in the documentation.

See more at https://github.com/apify/apify-client-js/actions/runs/32943962700#summary-98100614913

@barjin
barjin requested review from B4nan and janbuchar and removed request for janbuchar August 26, 2026 07:43
@barjin
barjin merged commit 7226376 into master Aug 26, 2026
8 checks passed
@barjin
barjin deleted the chore/remove-api-report-from-codeowners branch August 26, 2026 08:30
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

t-tooling Issues with this label are in the ownership of the tooling team.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants