Skip to content

Add missing Kinesis properties - #91

Open
purkhusid wants to merge 12 commits into
aws-controllers-k8s:mainfrom
purkhusid:update-kinesis-properties
Open

purkhusid wants to merge 12 commits into
aws-controllers-k8s:mainfrom
purkhusid:update-kinesis-properties

Conversation

@purkhusid

Copy link
Copy Markdown

What?

Adds the following properties

  • resourcePolicy
  • enforceConsumerDeletion
  • maxRecordSizeinKiB
  • shardLevelMetrics
  • warmThroughputMiBps

@ack-prow
ack-prow Bot requested review from knottnt and michaelhtm June 26, 2026 13:23
@ack-prow

ack-prow Bot commented Jun 26, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: purkhusid
Once this PR has been reviewed and has the lgtm label, please assign knottnt for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ack-prow

ack-prow Bot commented Jun 26, 2026

Copy link
Copy Markdown

Hi @purkhusid. Thanks for your PR.

I'm waiting for a aws-controllers-k8s member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ack-prow ack-prow Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Jun 26, 2026
@purkhusid

Copy link
Copy Markdown
Author

@michaelhtm I'll update this PR once #89 is merged

@purkhusid

Copy link
Copy Markdown
Author

@michaelhtm Any chance that teh e2e tests could be run so that I can verify them? I've been running a custom build with these changes and #89 in our clusters and it is working great.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant