This repository was archived by the owner on Jun 20, 2023. It is now read-only.
Docker build speedup - #114
Open
Tim Sogard (drags) wants to merge 1 commit into
Open
Conversation
Re-order docker file so that layers which are unlikely to change between builds appear first. Layers are now: clamav packages, pip requirements, and finally lambda code. This means that changing lambda code will not require re-installing clamav packages or rebuilding the rest of the image. Added a cleandocker target to the Makefile in order to explicitly rebuild from scratch (useful for when an update to clamav packages is required)
|
|
Ricardo MM (ricardomm85)
approved these changes
Mar 7, 2022
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to subscribe to this conversation on GitHub.
Already have an account?
Sign in.
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Small change to Dockerfile layer ordering. Layers which are expected to change infrequently appear earlier in the Dockerfile, reducing the time to
make archivefor code changes. Amake cleandockertarget has been added to the Makefile to explicitly delete the docker image when seeking to build from scratch.The original ordering may have been intentional (as a way of forcing fresh docker images on every code change), this is just intended to make development quicker. Feel free to close without verbose commentary if so.