Skip to content

chore: highlighter improvements and deps upgrade - #392

Merged
DylanPiercey merged 1 commit into
mainfrom
highlighter-improvements
Aug 13, 2025
Merged

chore: highlighter improvements and deps upgrade#392
DylanPiercey merged 1 commit into
mainfrom
highlighter-improvements

Conversation

@DylanPiercey

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 13, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6de6e94

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
marko-vscode Patch
@marko/language-server Patch
@marko/language-tools Patch
@marko/type-check Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@DylanPiercey
DylanPiercey merged commit ca1a4b5 into main Aug 13, 2025
4 checks passed
@DylanPiercey
DylanPiercey deleted the highlighter-improvements branch August 13, 2025 18:24
@github-actions github-actions Bot mentioned this pull request Aug 13, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant