Skip to content

fix: if tag scope hoisting - #398

Merged
DylanPiercey merged 1 commit into
mainfrom
fix-if-tag-hoisting
Aug 16, 2025
Merged

fix: if tag scope hoisting#398
DylanPiercey merged 1 commit into
mainfrom
fix-if-tag-hoisting

Conversation

@DylanPiercey

Copy link
Copy Markdown
Contributor

resolves #397

@changeset-bot

changeset-bot Bot commented Aug 16, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c90d769

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 4 packages
Name Type
@marko/language-server Patch
@marko/language-tools Patch
@marko/type-check Patch
marko-vscode Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@DylanPiercey
DylanPiercey merged commit 70bb404 into main Aug 16, 2025
4 checks passed
@DylanPiercey
DylanPiercey deleted the fix-if-tag-hoisting branch August 16, 2025 03:40
@github-actions github-actions Bot mentioned this pull request Aug 16, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Another incorrect typechecking when nested in <if>

1 participant