Skip to content

[release-4.21] OCPBUGS-76390: Fix framework initialization - #389

Closed
openshift-cherrypick-robot wants to merge 1 commit into
openshift:release-4.21from
openshift-cherrypick-robot:cherry-pick-384-to-release-4.21
Closed

[release-4.21] OCPBUGS-76390: Fix framework initialization#389
openshift-cherrypick-robot wants to merge 1 commit into
openshift:release-4.21from
openshift-cherrypick-robot:cherry-pick-384-to-release-4.21

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #384

/assign sunzhaohua2

@openshift-ci-robot

Copy link
Copy Markdown

@openshift-cherrypick-robot: Jira Issue OCPBUGS-74343 has been cloned as Jira Issue OCPBUGS-76390. Will retitle bug to link to clone.
/retitle [release-4.21] OCPBUGS-76390: Fix framework initialization

Details

In response to this:

This is an automated cherry-pick of #384

/assign sunzhaohua2

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot changed the title [release-4.21] OCPBUGS-74343: Fix framework initialization [release-4.21] OCPBUGS-76390: Fix framework initialization Feb 9, 2026
@openshift-ci-robot openshift-ci-robot added jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Feb 9, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@openshift-cherrypick-robot: This pull request references Jira Issue OCPBUGS-76390, which is invalid:

  • release note text must be set and not match the template OR release note type must be set to "Release Note Not Required". For more information you can reference the OpenShift Bug Process.

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

This is an automated cherry-pick of #384

/assign sunzhaohua2

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Feb 9, 2026
@openshift-ci
openshift-ci Bot requested review from RadekManak and nrb February 9, 2026 13:48

@damdo damdo left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 9, 2026
@openshift-ci

openshift-ci Bot commented Feb 9, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: damdo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 9, 2026
@neisw

neisw commented Feb 9, 2026

Copy link
Copy Markdown
Contributor

/hold

https://prow.ci.openshift.org/job-history/gs/test-platform-results/pr-logs/directory/pull-ci-openshift-origin-main-e2e-aws-ovn-microshift shows failures like

Failed to get info from binary cluster-control-plane-machine-set-operator-ext: failed running '/tmp/secret/.cache/openshift-tests/registry_build06_ci_openshift_org_ci-op-6t52im5m_release_sha256_52f4794cf89ad6430926d9a45a74533ce28b958d17b27ff489d0f7a1f32ab0fd_af343b20539b/cluster-control-plane-machine-set-operator-ext info': exit status 2\nOutput: panic: failed to initialize framework: failed to get infrastructure resource: unable to retrieve the complete list of server APIs: config.openshift.io/v1: no matches for config.openshift.io/v1, Resource=\n\ngoroutine 1

#384 will need to be reverted while this is worked out.

@neisw

neisw commented Feb 9, 2026

Copy link
Copy Markdown
Contributor

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 9, 2026
@openshift-ci

openshift-ci Bot commented Feb 9, 2026

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-ovn-etcd-scaling 5480b96 link true /test e2e-aws-ovn-etcd-scaling

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@sunzhaohua2

Copy link
Copy Markdown
Contributor

/close

In favor of #390

@openshift-ci openshift-ci Bot closed this Feb 28, 2026
@openshift-ci

openshift-ci Bot commented Feb 28, 2026

Copy link
Copy Markdown
Contributor

@sunzhaohua2: Closed this PR.

Details

In response to this:

/close

In favor of #390

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci-robot

Copy link
Copy Markdown

@openshift-cherrypick-robot: This pull request references Jira Issue OCPBUGS-76390. The bug has been updated to no longer refer to the pull request using the external bug tracker. All external bug links have been closed. The bug has been moved to the NEW state.

Details

In response to this:

This is an automated cherry-pick of #384

/assign sunzhaohua2

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. jira/severity-moderate Referenced Jira bug's severity is moderate for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants