Skip to content

[release-5.0] OCPBUGS-115467: Guard against empty workspace field on vsphere - #419

Open
openshift-cherrypick-robot wants to merge 1 commit into
openshift:release-5.0from
openshift-cherrypick-robot:cherry-pick-407-to-release-5.0
Open

[release-5.0] OCPBUGS-115467: Guard against empty workspace field on vsphere#419
openshift-cherrypick-robot wants to merge 1 commit into
openshift:release-5.0from
openshift-cherrypick-robot:cherry-pick-407-to-release-5.0

Conversation

@openshift-cherrypick-robot

Copy link
Copy Markdown

This is an automated cherry-pick of #407

/assign damdo

/cherrypick release-4.22 release-4.21 release-4.20

Signed-off-by: Nolan Brubaker <nolan@nbrubaker.com>
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@coderabbitai

coderabbitai Bot commented Sep 2, 2026

Copy link
Copy Markdown

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository: openshift/coderabbit/.coderabbit.yaml

Review profile: CHILL

Plan: Enterprise

Run ID: 5a21826f-c103-41a7-855f-1b0bc539c12f

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@openshift-ci openshift-ci Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 2, 2026
@openshift-ci
openshift-ci Bot requested review from nrb and racheljpg September 2, 2026 07:31
@openshift-ci-robot

Copy link
Copy Markdown

@openshift-cherrypick-robot: Jira Issue OCPBUGS-86988 has been cloned as Jira Issue OCPBUGS-115467. Will retitle bug to link to clone.
/retitle [release-5.0] OCPBUGS-115467: Guard against empty workspace field on vsphere

Details

In response to this:

This is an automated cherry-pick of #407

/assign damdo

/cherrypick release-4.22 release-4.21 release-4.20

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot changed the title [release-5.0] OCPBUGS-86988: Guard against empty workspace field on vsphere [release-5.0] OCPBUGS-115467: Guard against empty workspace field on vsphere Sep 2, 2026
@openshift-ci-robot openshift-ci-robot added jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Sep 2, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@openshift-cherrypick-robot: This pull request references Jira Issue OCPBUGS-115467, which is invalid:

  • release note text must be set and not match the template OR release note type must be set to "Release Note Not Required". For more information you can reference the OpenShift Bug Process.
  • expected dependent Jira Issue OCPBUGS-86988 to target a version in 5.1.0, but it targets "5.0" instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

This is an automated cherry-pick of #407

/assign damdo

/cherrypick release-4.22 release-4.21 release-4.20

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@damdo damdo left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
/lgtm

/verified by ci

/label backport-risk-assessed

@openshift-ci-robot openshift-ci-robot added the verified Signifies that the PR passed pre-merge verification criteria label Sep 2, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@damdo: This PR has been marked as verified by ci.

Details

In response to this:

/approve
/lgtm

/verified by ci

/label backport-risk-assessed

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci openshift-ci Bot added backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. lgtm Indicates that a PR is ready to be merged. labels Sep 2, 2026
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling tests matching the pipeline_run_if_changed or not excluded by pipeline_skip_if_only_changed parameters:
/test e2e-azure-operator
/test e2e-azure-ovn-etcd-scaling
/test e2e-gcp-operator
/test e2e-gcp-ovn-etcd-scaling
/test e2e-nutanix-ovn
/test e2e-openstack-operator
/test e2e-vsphere-multi-vcenter-operator
/test e2e-vsphere-multi-zone-operator
/test e2e-vsphere-operator
/test e2e-vsphere-operator-periodic-pre
/test e2e-vsphere-operator-periodic-pre-techpreview
/test e2e-vsphere-ovn-etcd-scaling
/test e2e-aws-operator
/test e2e-aws-operator-techpreview
/test e2e-aws-ovn
/test e2e-aws-ovn-etcd-scaling
/test e2e-aws-ovn-serial-1of2
/test e2e-aws-ovn-serial-2of2
/test e2e-aws-ovn-upgrade

@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: damdo

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 2, 2026
@damdo

damdo commented Sep 2, 2026

Copy link
Copy Markdown
Member

/override ci/prow/e2e-vsphere-ovn-etcd-scaling ci/prow/e2e-azure-ovn-etcd-scaling ci/prow/e2e-gcp-ovn-etcd-scaling ci/prow/e2e-aws-ovn-etcd-scaling

@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

@damdo: Overrode contexts on behalf of damdo: ci/prow/e2e-aws-ovn-etcd-scaling, ci/prow/e2e-azure-ovn-etcd-scaling, ci/prow/e2e-gcp-ovn-etcd-scaling, ci/prow/e2e-vsphere-ovn-etcd-scaling

Details

In response to this:

/override ci/prow/e2e-vsphere-ovn-etcd-scaling ci/prow/e2e-azure-ovn-etcd-scaling ci/prow/e2e-gcp-ovn-etcd-scaling ci/prow/e2e-aws-ovn-etcd-scaling

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@damdo

damdo commented Sep 2, 2026

Copy link
Copy Markdown
Member

/test unit

@damdo

damdo commented Sep 2, 2026

Copy link
Copy Markdown
Member

/override ci/prow/e2e-vsphere-ovn-etcd-scaling ci/prow/e2e-azure-ovn-etcd-scaling ci/prow/e2e-gcp-ovn-etcd-scaling ci/prow/e2e-aws-ovn-etcd-scaling

@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

@damdo: Overrode contexts on behalf of damdo: ci/prow/e2e-aws-ovn-etcd-scaling, ci/prow/e2e-azure-ovn-etcd-scaling, ci/prow/e2e-gcp-ovn-etcd-scaling, ci/prow/e2e-vsphere-ovn-etcd-scaling

Details

In response to this:

/override ci/prow/e2e-vsphere-ovn-etcd-scaling ci/prow/e2e-azure-ovn-etcd-scaling ci/prow/e2e-gcp-ovn-etcd-scaling ci/prow/e2e-aws-ovn-etcd-scaling

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@damdo

damdo commented Sep 2, 2026

Copy link
Copy Markdown
Member

/retest

@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown
Contributor

@openshift-cherrypick-robot: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-gcp-ovn-etcd-scaling c9b983b link false /test e2e-gcp-ovn-etcd-scaling
ci/prow/e2e-aws-ovn-etcd-scaling c9b983b link true /test e2e-aws-ovn-etcd-scaling
ci/prow/e2e-vsphere-operator-periodic-pre c9b983b link false /test e2e-vsphere-operator-periodic-pre
ci/prow/e2e-vsphere-ovn-etcd-scaling c9b983b link false /test e2e-vsphere-ovn-etcd-scaling
ci/prow/e2e-vsphere-operator-periodic-pre-techpreview c9b983b link false /test e2e-vsphere-operator-periodic-pre-techpreview
ci/prow/e2e-azure-ovn-etcd-scaling c9b983b link false /test e2e-azure-ovn-etcd-scaling

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files. verified Signifies that the PR passed pre-merge verification criteria

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants