Skip to content

Ensure we can suspend from a given component more than once - #436

Merged
JoviDeCroock merged 1 commit into
mainfrom
stream-multi-throw
Dec 12, 2025
Merged

Ensure we can suspend from a given component more than once#436
JoviDeCroock merged 1 commit into
mainfrom
stream-multi-throw

Conversation

@JoviDeCroock

@JoviDeCroock JoviDeCroock commented Dec 12, 2025

Copy link
Copy Markdown
Member

Resolves #435

The issue was that we start rendering again from the same component which means there was no outer try-catch block to rely on. To fix this we don't start rendering from the parent, else we might break useId allocation, we resume from the same child but make onError recursive.

@changeset-bot

changeset-bot Bot commented Dec 12, 2025

Copy link
Copy Markdown

馃 Changeset detected

Latest commit: 6bce3dd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
preact-render-to-string Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@marvinhagemeister marvinhagemeister left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@JoviDeCroock
JoviDeCroock merged commit 80ba7dc into main Dec 12, 2025
1 check passed
@JoviDeCroock
JoviDeCroock deleted the stream-multi-throw branch December 12, 2025 08:12
@github-actions github-actions Bot mentioned this pull request Dec 12, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

renderToReadableStream fails when you suspend a single component two times

2 participants