Skip to content

fix: Correct semver ranges for v11 beta releases - #585

Merged
JoviDeCroock merged 1 commit into
mainfrom
fix/correct-semver-for-v11
Aug 20, 2025
Merged

fix: Correct semver ranges for v11 beta releases#585
JoviDeCroock merged 1 commit into
mainfrom
fix/correct-semver-for-v11

Conversation

@rschristian

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 20, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7ddbae9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@prefresh/web-dev-server Patch
@prefresh/webpack Patch
@prefresh/nollup Patch
@prefresh/core Patch
@prefresh/vite Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@JoviDeCroock
JoviDeCroock merged commit c1f1372 into main Aug 20, 2025
1 check passed
@JoviDeCroock
JoviDeCroock deleted the fix/correct-semver-for-v11 branch August 20, 2025 05:47
@github-actions github-actions Bot mentioned this pull request Aug 20, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants