Skip to content

Changing batch to use a single flag instead of a counter - #143

Merged
marvinhagemeister merged 2 commits into
preactjs:mainfrom
Pauan:improving-signals-1
Sep 14, 2022
Merged

Changing batch to use a single flag instead of a counter#143
marvinhagemeister merged 2 commits into
preactjs:mainfrom
Pauan:improving-signals-1

Conversation

@Pauan

@Pauan Pauan commented Sep 14, 2022

Copy link
Copy Markdown
Contributor

As discussed in #104, I'm going to try and split the improvements into multiple smaller PRs.

All of the tests pass, but this does slightly change the behavior, so I hope it doesn't break anybody's code.

@changeset-bot

changeset-bot Bot commented Sep 14, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f2ba3d6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@preact/signals-core Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@netlify

netlify Bot commented Sep 14, 2022

Copy link
Copy Markdown

Deploy Preview for preact-signals-demo ready!

Name Link
🔨 Latest commit f2ba3d6
🔍 Latest deploy log https://app.netlify.com/sites/preact-signals-demo/deploys/632176663953b4000924da76
😎 Deploy Preview https://deploy-preview-143--preact-signals-demo.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site settings.

@marvinhagemeister marvinhagemeister left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Awesome, this is great! Thank you for the PR 👍

@marvinhagemeister
marvinhagemeister merged commit 0c2908a into preactjs:main Sep 14, 2022
@Pauan
Pauan deleted the improving-signals-1 branch September 14, 2022 07:15
@github-actions github-actions Bot mentioned this pull request Sep 15, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants