Skip to content

Add missing changeset for #130 - #146

Merged
marvinhagemeister merged 1 commit into
mainfrom
changeset
Sep 14, 2022
Merged

Add missing changeset for #130#146
marvinhagemeister merged 1 commit into
mainfrom
changeset

Conversation

@marvinhagemeister

@marvinhagemeister marvinhagemeister commented Sep 14, 2022

Copy link
Copy Markdown
Member

Apologies, forgot to add it when merging #130

@changeset-bot

changeset-bot Bot commented Sep 14, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9e798fd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@preact/signals-react Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@netlify

netlify Bot commented Sep 14, 2022

Copy link
Copy Markdown

Deploy Preview for preact-signals-demo ready!

Name Link
🔨 Latest commit 9e798fd
🔍 Latest deploy log https://app.netlify.com/sites/preact-signals-demo/deploys/6321c2baa6c65a000814fc99
😎 Deploy Preview https://deploy-preview-146--preact-signals-demo.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site settings.

@github-actions

Copy link
Copy Markdown
Contributor

Size Change: 0 B

Total Size: 65.1 kB

ℹ️ View Unchanged
Filename Size
docs/dist/assets/client.********.js 45.8 kB
docs/dist/assets/index.********.js 1.06 kB
docs/dist/assets/signals-core.module.********.js 932 B
docs/dist/assets/signals.module.********.js 5.67 kB
docs/dist/assets/style.********.js 21 B
docs/dist/assets/style.*********.js 21 B
docs/dist/assets/style.********.css 994 B
docs/dist/basic-********.js 226 B
docs/dist/demos-********.js 3.45 kB
docs/dist/nesting-********.js 1.11 kB
docs/dist/react-********.js 236 B
packages/core/dist/signals-core.js 984 B
packages/core/dist/signals-core.mjs 788 B
packages/preact/dist/signals.js 1.13 kB
packages/preact/dist/signals.mjs 1.11 kB
packages/react/dist/signals.js 814 B
packages/react/dist/signals.mjs 773 B

compressed-size-action

@marvinhagemeister
marvinhagemeister merged commit 03f72fd into main Sep 14, 2022
@marvinhagemeister
marvinhagemeister deleted the changeset branch September 14, 2022 14:42
@github-actions github-actions Bot mentioned this pull request Sep 15, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant