Skip to content

Fix React.lazy and React.forwardRef component updates tracking. - #427

Closed
XantreDev wants to merge 6 commits into
preactjs:react-1.2.xfrom
XantreDev:fix-lazy-reactive
Closed

Fix React.lazy and React.forwardRef component updates tracking.#427
XantreDev wants to merge 6 commits into
preactjs:react-1.2.xfrom
XantreDev:fix-lazy-reactive

Conversation

@XantreDev

@XantreDev XantreDev commented Oct 14, 2023

Copy link
Copy Markdown
Contributor

Based on @ywang1724 pr #342. Appreciate for help.

Added tests of reactiveness for React.lazy and React.forwardRef and made program to pass it.
Updated pnpm-lock while installation, because ci failed

@changeset-bot

changeset-bot Bot commented Oct 14, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: cf51a22

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@preact/signals-react Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@netlify

netlify Bot commented Oct 14, 2023

Copy link
Copy Markdown

Deploy Preview for preact-signals-demo ready!

Name Link
🔨 Latest commit ae15b16
🔍 Latest deploy log https://app.netlify.com/sites/preact-signals-demo/deploys/652a4c58d3b2880008214fd4
😎 Deploy Preview https://deploy-preview-427--preact-signals-demo.netlify.app
📱 Preview on mobile
Toggle QR Code...

QR Code

Use your smartphone camera to open QR code link.

To edit notification comments on pull requests, go to your Netlify site configuration.

@XantreDev

Copy link
Copy Markdown
Contributor Author

@marvinhagemeister @JoviDeCroock Review me please). Its fix for 1.2.1 version, because its most stable one and for now the only one working in react native.
#301 #420

@JoviDeCroock JoviDeCroock left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Seems to have a lot of merge conflicts and non-related changes in the lockfile

Comment thread pnpm-lock.yaml Outdated
@@ -1,243 +1,308 @@
lockfileVersion: 5.4
lockfileVersion: '6.0'

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I don't think adding the lockfile changes is in scope of this PR

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Sorry but without this changes pipelines was failed

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Then we need to update/alter the pipeline or fix it separately.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Should I cherry pick and provide another PR? In which branch i should merge it?

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@rschristian What should I do?

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@rschristian

Copy link
Copy Markdown
Member

Its fix for 1.2.1 version, because its most stable one and for now the only one working in react native.

Am I understanding you correctly in that this is based on 1.2.1, instead of the branch head? Definitely can't merge this into main if so.

Precisely, is it 1.2.1 or 1.2.2? Should probably be the latter and we can make a branch for you to merge into if needs be.

@XantreDev

Copy link
Copy Markdown
Contributor Author

I think it should be in another branch, this changes for 1.2.2

@XantreDev

Copy link
Copy Markdown
Contributor Author

I think we can cherry pick commit related with adding tests for forwardRef and lazy. But other things merge into separate branch.

@XantreDev

Copy link
Copy Markdown
Contributor Author

@JoviDeCroock @rschristian what should i do to release 1.2.3 with this fixes? I want to fix this bug, because it causes problems in my project

@rschristian
rschristian changed the base branch from main to react-old November 5, 2023 10:33
@rschristian

rschristian commented Nov 5, 2023

Copy link
Copy Markdown
Member

Sorry for the delay, just been incredibly busy and missed this. I created a new branch from 1.2.2 and switched this to target it. You might want to take a look at the pnpm.lock file though, that diff looks odd and unnecessary. Edit: Just reread prior comments -- revert the change entirely. If it's needed for the CI or the CI needs alterations, that can be done in another PR.

I don't use React and therefore this package, so I can't really comment here. @andrewiggins is probably the best person to comment on this (if he has the time)

@rschristian
rschristian changed the base branch from react-old to react-1.2.x November 5, 2023 10:42
@XantreDev

Copy link
Copy Markdown
Contributor Author

#431 Extracted lockfile changes to separate PR

@rschristian

Copy link
Copy Markdown
Member

The CI being limited to PRs against main is one way to fix the issue 😂

@aspizu

aspizu commented Jan 2, 2024

Copy link
Copy Markdown

any updates on this?

@XantreDev

Copy link
Copy Markdown
Contributor Author

any updates on this?

You can use the 2.0.0 version with the babel plugin now

@XantreDev XantreDev closed this Sep 22, 2024
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants