Skip to content

Pr/1180 - #1195

Merged
steveukx merged 6 commits into
mainfrom
pr/1180
Sep 25, 2026
Merged

Pr/1180#1195
steveukx merged 6 commits into
mainfrom
pr/1180

Conversation

@steveukx

@steveukx steveukx commented Sep 22, 2026 •

Copy link
Copy Markdown
Owner

Scheduled for >=4.0.1

Closes #1180

@changeset-bot

changeset-bot Bot commented Sep 22, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 31c981d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 7 packages
Name Type
simple-git Patch
@simple-git/test-javascript-cjs-consumer Patch
@simple-git/test-javascript-esm-consumer Patch
@simple-git/test-typescript-cjs-consumer Patch
@simple-git/test-typescript-esm-consumer Patch
@simple-git/test-typescript-strict-cjs-consumer Patch
@simple-git/test-typescript-strict-esm-consumer Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@steveukx
steveukx merged commit 76f308e into main Sep 25, 2026
11 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants